Skip to content

SOV-4642: Table mobile layout break - #1096

Merged
creed-victor merged 3 commits into
developfrom
fix/mobileTable
Sep 8, 2025
Merged

creed-victor merged 3 commits into
developfrom
fix/mobileTable

Conversation

@norugpull

Copy link
Copy Markdown
Collaborator

@norugpull norugpull self-assigned this Aug 13, 2025
@norugpull norugpull added the bug Something isn't working label Aug 13, 2025
@changeset-bot

changeset-bot Bot commented Aug 13, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4f107d7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@sovryn/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@netlify

netlify Bot commented Aug 13, 2025 •

Copy link
Copy Markdown

✅ Deploy Preview for sovryn-storybook ready!

Name Link
🔨 Latest commit 4f107d7
🔍 Latest deploy log https://app.netlify.com/projects/sovryn-storybook/deploys/689dfbe128ac23000837ec54
😎 Deploy Preview https://deploy-preview-1096--sovryn-storybook.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@netlify

netlify Bot commented Aug 13, 2025 •

Copy link
Copy Markdown

✅ Deploy Preview for sovryn-dapp ready!

Name Link
🔨 Latest commit 4f107d7
🔍 Latest deploy log https://app.netlify.com/projects/sovryn-dapp/deploys/689dfbe1bea0130008b87e01
😎 Deploy Preview https://deploy-preview-1096.preview.sovryns.dev
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@tiltom
tiltom self-requested a review August 13, 2025 13:46
@tiltom tiltom added this to the v1.1.32 milestone Sep 4, 2025
@creed-victor
creed-victor merged commit d98acd1 into develop Sep 8, 2025
13 checks passed
@creed-victor
creed-victor deleted the fix/mobileTable branch September 8, 2025 06:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

bug Something isn't working

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants