Skip to content

SOV-5105: Fix gas estimation error - #1101

Merged
creed-victor merged 3 commits into
developfrom
fix/SOV-5105-zero-redemptions
Sep 8, 2025
Merged

creed-victor merged 3 commits into
developfrom
fix/SOV-5105-zero-redemptions

Conversation

@tiltom

@tiltom tiltom commented Aug 28, 2025

Copy link
Copy Markdown
Collaborator

@tiltom
tiltom requested a review from creed-victor August 28, 2025 09:19
@tiltom tiltom self-assigned this Aug 28, 2025
@tiltom tiltom added the bug Something isn't working label Aug 28, 2025
@changeset-bot

changeset-bot Bot commented Aug 28, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 952c05a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@sovryn/sdk Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@netlify

netlify Bot commented Aug 28, 2025 •

Copy link
Copy Markdown

✅ Deploy Preview for sovryn-storybook canceled.

Name Link
🔨 Latest commit 952c05a
🔍 Latest deploy log https://app.netlify.com/projects/sovryn-storybook/deploys/68b70b876a84c60008bbeb57

@netlify

netlify Bot commented Aug 28, 2025 •

Copy link
Copy Markdown

✅ Deploy Preview for sovryn-dapp ready!

Name Link
🔨 Latest commit 952c05a
🔍 Latest deploy log https://app.netlify.com/projects/sovryn-dapp/deploys/68b70b873854e2000836c3ad
😎 Deploy Preview https://deploy-preview-1101.preview.sovryns.dev
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@tiltom tiltom added this to the v1.1.32 milestone Sep 4, 2025
@creed-victor
creed-victor merged commit fd20140 into develop Sep 8, 2025
11 checks passed
@creed-victor
creed-victor deleted the fix/SOV-5105-zero-redemptions branch September 8, 2025 06:50
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

bug Something isn't working

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants