Skip to content

allmark: init at 0.10.0 - #207781

Merged
viraptor merged 1 commit into
NixOS:masterfrom
urandom2:allmark
Dec 27, 2022
Merged

allmark: init at 0.10.0#207781
viraptor merged 1 commit into
NixOS:masterfrom
urandom2:allmark

Conversation

@urandom2

Copy link
Copy Markdown
Contributor
Description of changes

We are using buildGoPackage, because upstream is not a module: see andreaskoch/allmark#36.

Fixes #206256

Things done
  • Built on platform(s)
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • For non-Linux: Is sandbox = true set in nix.conf? (See Nix manual)
  • Tested, as applicable:
  • Tested compilation of all packages that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage
  • Tested basic functionality of all binary files (usually in ./result/bin/)
  • 23.05 Release Notes (or backporting 22.11 Release notes)
    • (Package updates) Added a release notes entry if the change is major or breaking
    • (Module updates) Added a release notes entry if the change is significant
    • (Module addition) Added a release notes entry if adding a new NixOS module
    • (Release notes changes) Ran nixos/doc/manual/md-to-db.sh to update generated release notes
  • Fits CONTRIBUTING.md.

@ofborg ofborg Bot added 8.has: package (new) This PR adds a new package 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. labels Dec 26, 2022
@nixos-discourse

Copy link
Copy Markdown

This pull request has been mentioned on NixOS Discourse. There might be relevant details there:

https://discourse.nixos.org/t/prs-ready-for-review/3032/1601

Comment thread pkgs/servers/allmark/default.nix Outdated

@viraptor viraptor left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nitpick about formatting. Otherwise seems ok.

Comment thread pkgs/servers/allmark/default.nix Outdated
We are using buildGoPackage, because upstream is not a module: see
andreaskoch/allmark#36.

Fixes NixOS#206256
@viraptor
viraptor merged commit 5ecdcd6 into NixOS:master Dec 27, 2022
@bbigras

bbigras commented Dec 29, 2022

Copy link
Copy Markdown
Contributor

Btw, the https://allmark.io homepage seems wrong. I got a message saying that the site is gone.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

8.has: package (new) This PR adds a new package 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Allmark

5 participants