Skip to content

feat: add tenant id - #6

Merged
trivikr merged 3 commits into
awslabs:mainfrom
maxday:maxday/add-tenant-id
Sep 11, 2025
Merged

trivikr merged 3 commits into
awslabs:mainfrom
maxday:maxday/add-tenant-id

Conversation

@maxday

@maxday maxday commented Sep 11, 2025

Copy link
Copy Markdown
Contributor

Description of changes:

By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.

@maxday
maxday requested a review from a team as a code owner September 11, 2025 16:37
@changeset-bot

changeset-bot Bot commented Sep 11, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c0029ca

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@aws/lambda-invoke-store Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@trivikr trivikr left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Two changes needed:

  • Update API Reference in README to mention the API
  • Run yarn changeset to add changeset for a minor version

@maxday

maxday commented Sep 11, 2025

Copy link
Copy Markdown
Contributor Author
  • Run yarn changeset to add changeset for a minor version

done, I would suggest you to add this the the README.md or CONTRIBUTING.md if you don't want to repeat this for every PRs

@trivikr
trivikr merged commit 83fa470 into awslabs:main Sep 11, 2025
3 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants