Skip to content

Support json logging format - #1256

Merged
k8s-ci-robot merged 1 commit into
kubernetes-csi:masterfrom
huww98:json-log
Feb 28, 2025
Merged

k8s-ci-robot merged 1 commit into
kubernetes-csi:masterfrom
huww98:json-log

Conversation

@huww98

@huww98 huww98 commented Dec 21, 2024 •

Copy link
Copy Markdown
Contributor

What type of PR is this?

/kind feature

What this PR does / why we need it:

Only one line of log is changed to structured logging, to prove the json format works. We can migrate others in the future.

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Like: kubernetes-csi/external-resizer#330

Related:

Before:

I0228 16:17:28.677813   69383 main.go:169] Version: unknown
E0228 16:17:28.678324   69383 main.go:174] unable to load in-cluster configuration, KUBERNETES_SERVICE_HOST and KUBERNETES_SERVICE_PORT must be defined

After:

I0228 16:18:55.410566   69629 main.go:179] "Version" version="unknown"
E0228 16:18:55.410944   69629 main.go:184] unable to load in-cluster configuration, KUBERNETES_SERVICE_HOST and KUBERNETES_SERVICE_PORT must be defined

After and with --logging-format=json

{"ts":1740730756069.866,"caller":"snapshot-controller/main.go:179","msg":"Version","v":0,"version":"unknown"}
{"ts":1740730756070.206,"caller":"snapshot-controller/main.go:184","msg":"unable to load in-cluster configuration, KUBERNETES_SERVICE_HOST and KUBERNETES_SERVICE_PORT must be defined"}

Does this PR introduce a user-facing change?:

Add support for -logging-format=json
Remove klog specific flags according to KEP-2845

KEP-2845: Deprecate klog specific flags in Kubernetes Components

@k8s-ci-robot k8s-ci-robot added release-note Denotes a PR that will be considered when it comes time to generate release notes. kind/feature Categorizes issue or PR as related to a new feature. labels Dec 21, 2024
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Dec 21, 2024
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Hi @huww98. Thanks for your PR.

I'm waiting for a kubernetes-csi member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot k8s-ci-robot added needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Dec 21, 2024
@xing-yang

Copy link
Copy Markdown
Collaborator

/ok-to-test

@k8s-ci-robot k8s-ci-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Dec 23, 2024
@xing-yang

Copy link
Copy Markdown
Collaborator

Can you replace "KEP-2845" in the release notes with a description of what that feature is about?
Also add "KEP-2845" and a link to the KEP in the PR description itself.

@huww98

huww98 commented Dec 23, 2024

Copy link
Copy Markdown
Contributor Author

Can you replace "KEP-2845" in the release notes with a description of what that feature is about?

Actually already included, that is "Remove klog specific flags". KEP-2845 is just offering more context to the interested reader of the release note. kubernetes-csi/external-resizer#411 contains some more discussion about this change.

Also add "KEP-2845" and a link to the KEP in the PR description itself.

Done

@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Feb 26, 2025
@k8s-ci-robot k8s-ci-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Feb 27, 2025
@huww98

huww98 commented Feb 27, 2025

Copy link
Copy Markdown
Contributor Author

@xing-yang Rebased. I think this is the only CSI sidecar that don't support json logging now. Can we merge this so that we can use json logging for the whole CSI driver?

@xing-yang

Copy link
Copy Markdown
Collaborator

In the PR description, can you add output of the logs before and after your change?

@huww98

huww98 commented Feb 28, 2025

Copy link
Copy Markdown
Contributor Author

In the PR description, can you add output of the logs before and after your change?

Done

@xing-yang

Copy link
Copy Markdown
Collaborator

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 28, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: huww98, xing-yang

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 28, 2025
@k8s-ci-robot
k8s-ci-robot merged commit 59d7297 into kubernetes-csi:master Feb 28, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/feature Categorizes issue or PR as related to a new feature. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note Denotes a PR that will be considered when it comes time to generate release notes. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants