Skip to content

Test post-modifications upstream - #436

Closed
nforro wants to merge 1 commit into
mainfrom
debug
Closed

nforro wants to merge 1 commit into
mainfrom
debug

Conversation

@nforro

@nforro nforro commented Dec 2, 2024

Copy link
Copy Markdown
Member

No description provided.

@packit-as-a-service

Copy link
Copy Markdown

Failed to load packit config file:

Cannot parse package config. ValidationError({'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}}), 'jobs': {0: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 2: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 4: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 6: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 8: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 10: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 11: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 14: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 17: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}}})

For more info, please check out the documentation or contact the Packit team. You can also use our CLI command validate-config or our pre-commit hooks for validation of the configuration.

@packit-as-a-service

Copy link
Copy Markdown

Failed to load packit config file:

Cannot parse package config. ValidationError({'jobs': {0: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 2: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 4: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 6: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 8: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 10: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 11: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 14: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 17: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}}, 'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})})

For more info, please check out the documentation or contact the Packit team. You can also use our CLI command validate-config or our pre-commit hooks for validation of the configuration.

1 similar comment
@packit-as-a-service

Copy link
Copy Markdown

Failed to load packit config file:

Cannot parse package config. ValidationError({'jobs': {0: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 2: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 4: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 6: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 8: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 10: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 11: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 14: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 17: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}}, 'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})})

For more info, please check out the documentation or contact the Packit team. You can also use our CLI command validate-config or our pre-commit hooks for validation of the configuration.

@softwarefactory-project-zuul

Copy link
Copy Markdown
Contributor

Build succeeded.
https://softwarefactory-project.io/zuul/t/packit-service/buildset/85bdca48f17e403aaadd9a7fb83b470a

✔️ pre-commit SUCCESS in 6m 40s
✔️ specfile-tests-rpm-deps SUCCESS in 3m 05s
✔️ specfile-tests-pip-deps SUCCESS in 7m 11s

Signed-off-by: Nikola Forró <nforro@redhat.com>
@packit-as-a-service

Copy link
Copy Markdown

Failed to load packit config file:

Cannot parse package config. ValidationError({'jobs': {0: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 2: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 4: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 6: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 8: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 10: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 11: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 14: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 17: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}}, 'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})})

For more info, please check out the documentation or contact the Packit team. You can also use our CLI command validate-config or our pre-commit hooks for validation of the configuration.

2 similar comments
@packit-as-a-service

Copy link
Copy Markdown

Failed to load packit config file:

Cannot parse package config. ValidationError({'jobs': {0: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 2: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 4: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 6: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 8: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 10: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 11: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 14: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 17: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}}, 'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})})

For more info, please check out the documentation or contact the Packit team. You can also use our CLI command validate-config or our pre-commit hooks for validation of the configuration.

@packit-as-a-service

Copy link
Copy Markdown

Failed to load packit config file:

Cannot parse package config. ValidationError({'jobs': {0: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 2: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 4: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 6: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 8: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 10: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 11: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 14: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}, 17: {'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})}}, 'packages': defaultdict(<class 'dict'>, {'specfile': {'value': {'actions': ["Unknown action(s) provided: ['post-modifications']"]}}})})

For more info, please check out the documentation or contact the Packit team. You can also use our CLI command validate-config or our pre-commit hooks for validation of the configuration.

@nforro nforro closed this Dec 2, 2024
@nforro
nforro deleted the debug branch December 2, 2024 10:38
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant